font-manager: fix build issue - #557155
Merged
Merged
Conversation
ShowhyT
force-pushed
the
pkgs/font-manager
branch
2 times, most recently
from
August 27, 2026 20:38
960976c to
4b9be75
Compare
This comment was marked as outdated.
This comment was marked as outdated.
thunze
reviewed
Aug 27, 2026
ShowhyT
force-pushed
the
pkgs/font-manager
branch
2 times, most recently
from
August 28, 2026 09:20
af2d582 to
6cf9465
Compare
Member
Author
✅
|
thunze
approved these changes
Aug 28, 2026
thunze
left a comment
Member
There was a problem hiding this comment.
- Diff LGTM
- Verified that the package builds on
x86_64-linux - Verified basic functionality of the package on
x86_64-linux
Thanks!
Member
|
Duplicate of #556842 |
amarshall
reviewed
Aug 28, 2026
ShowhyT
force-pushed
the
pkgs/font-manager
branch
from
August 28, 2026 14:13
6cf9465 to
460cb81
Compare
ShowhyT
force-pushed
the
pkgs/font-manager
branch
from
August 28, 2026 16:13
a1a8858 to
0658f7b
Compare
Contributor
|
Successfully created backport PR for |
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Closes: #557142
Fix build issue with newer vala versions via a patch. Fetched the patch from this PR:
FontManager/font-manager#468
Things done
passthru.tests.nixpkgs-reviewon this PR. See nixpkgs-review usage../result/bin/.