font-manager: fix build - #556842
Closed
uninsane wants to merge 1 commit into
Closed
Conversation
fixes `nix-build -A font-manager` from x86_64-linux.
4 tasks
13 tasks
thunze
approved these changes
Aug 28, 2026
thunze
left a comment
Member
There was a problem hiding this comment.
Not sure what's going on with CI, but the changes LGTM.
- Diff LGTM
- Verified that the package builds on
x86_64-linux - Verified basic functionality of the package on
x86_64-linux
Thanks!
Member
|
#557155 was merged first. Thanks regardless. |
Contributor
Although this one is older, I merged the other because of the CI tests failures here. |
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
fixes
nix-build -A font-managerfrom x86_64-linux.Things done
passthru.tests.nixpkgs-reviewon this PR. See nixpkgs-review usage../result/bin/.