Skip to content

⬆️ Bump @storybook/react from 8.6.12 to 9.0.3 - #3846

Closed
dependabot[bot] wants to merge 1 commit into
mainfrom
dependabot/npm_and_yarn/storybook/react-9.0.3
Closed

⬆️ Bump @storybook/react from 8.6.12 to 9.0.3#3846
dependabot[bot] wants to merge 1 commit into
mainfrom
dependabot/npm_and_yarn/storybook/react-9.0.3

Conversation

@dependabot

@dependabot dependabot Bot commented on behalf of github Jun 1, 2025

Copy link
Copy Markdown
Contributor

Bumps @storybook/react from 8.6.12 to 9.0.3.

Release notes

Sourced from @​storybook/react's releases.

v9.0.3

9.0.3

v9.0.2

9.0.2

  • CLI: Respect --skip-install in postinstall scripts - #31605, thanks @​yannbf!

v9.0.1

9.0.1

v9.0.0

9.0.0

Storybook 9.0 is here

This is a huge release focused on testing and bundle size.

  • Component testing
    • 👆 Interactions
    • ♿️ Accessibility
    • 👁️ Visual changes
    • 🛡️ Coverage
  • 🪶 48% lighter bundle
  • 🏷️ Tags-based organization
  • 🌐 Story globals
  • 🏗️ Major upgrades: Svelte, Next, React Native, Angular

Please checkout our Migration guide to upgrade from earlier versions of Storybook. To see a comprehensive list of changes that went into 9.0, you can refer to the 9.0 prerelease changelogs

... (truncated)

Changelog

Sourced from @​storybook/react's changelog.

9.0.3

9.0.2

  • CLI: Respect --skip-install in postinstall scripts - #31605, thanks @​yannbf!

9.0.1

9.0.0

Storybook 9.0 is here

This is a huge release focused on testing and bundle size.

  • Component testing
    • 👆 Interactions
    • ♿️ Accessibility
    • 👁️ Visual changes
    • 🛡️ Coverage
  • 🪶 48% lighter bundle
  • 🏷️ Tags-based organization
  • 🌐 Story globals
  • 🏗️ Major upgrades: Svelte, Next, React Native, Angular

Please checkout our Migration guide to upgrade from earlier versions of Storybook. To see a comprehensive list of changes that went into 9.0, you can refer to the 9.0 prerelease changelogs

... (truncated)

Commits
  • 7cbe162 Bump version from "9.0.2" to "9.0.3" [skip ci]
  • 371930b Bump version from "9.0.1" to "9.0.2" [skip ci]
  • 9879693 Bump version from "9.0.0" to "9.0.1" [skip ci]
  • 5dd81ae Bump version from "9.0.0-rc.5" to "9.0.0" [skip ci]
  • fd3356c Bump version from "9.0.0-rc.4" to "9.0.0-rc.5" [skip ci]
  • b7afb17 Merge branch 'next' into docs-update-testing-section
  • 0f1d97b Bump version from "9.0.0-rc.3" to "9.0.0-rc.4" [skip ci]
  • 77b810e Bump version from "9.0.0-rc.2" to "9.0.0-rc.3" [skip ci]
  • 54f328c fix unit tests
  • 9d1831a Merge branch 'next' into docs-update-testing-section
  • Additional commits viewable in compare view

Dependabot compatibility score

Dependabot will resolve any conflicts with this PR as long as you don't alter it yourself. You can also trigger a rebase manually by commenting @dependabot rebase.


Dependabot commands and options

You can trigger Dependabot actions by commenting on this PR:

  • @dependabot rebase will rebase this PR
  • @dependabot recreate will recreate this PR, overwriting any edits that have been made to it
  • @dependabot merge will merge this PR after your CI passes on it
  • @dependabot squash and merge will squash and merge this PR after your CI passes on it
  • @dependabot cancel merge will cancel a previously requested merge and block automerging
  • @dependabot reopen will reopen this PR if it is closed
  • @dependabot close will close this PR and stop Dependabot recreating it. You can achieve the same result by closing it manually
  • @dependabot show <dependency name> ignore conditions will show all of the ignore conditions of the specified dependency
  • @dependabot ignore this major version will close this PR and stop Dependabot creating any more for this major version (unless you reopen the PR or upgrade to it yourself)
  • @dependabot ignore this minor version will close this PR and stop Dependabot creating any more for this minor version (unless you reopen the PR or upgrade to it yourself)
  • @dependabot ignore this dependency will close this PR and stop Dependabot creating any more for this dependency (unless you reopen the PR or upgrade to it yourself)

Bumps [@storybook/react](https://github.com/storybookjs/storybook/tree/HEAD/code/renderers/react) from 8.6.12 to 9.0.3.
- [Release notes](https://github.com/storybookjs/storybook/releases)
- [Changelog](https://github.com/storybookjs/storybook/blob/next/CHANGELOG.md)
- [Commits](https://github.com/storybookjs/storybook/commits/v9.0.3/code/renderers/react)

---
updated-dependencies:
- dependency-name: "@storybook/react"
  dependency-version: 9.0.3
  dependency-type: direct:development
  update-type: version-update:semver-major
...

Signed-off-by: dependabot[bot] <support@github.com>
@dependabot dependabot Bot added the dependencies Pull requests that update a dependency file label Jun 1, 2025
@dependabot
dependabot Bot requested review from OAGr and berekuk as code owners June 1, 2025 14:36
@github-project-automation github-project-automation Bot moved this to 🆕 To prioritize in Overview Jun 1, 2025
@changeset-bot

changeset-bot Bot commented Jun 1, 2025

Copy link
Copy Markdown

⚠️ No Changeset found

Latest commit: 6f80ec7

Merging this PR will not cause a version bump for any packages. If these changes should not result in a new version, you're good to go. If these changes should result in a version bump, you need to add a changeset.

This PR includes no changesets

When changesets are added to this PR, you'll see the packages that this PR includes changesets for and the associated semver types

Click here to learn what changesets are, and how to add one.

Click here if you're a maintainer who wants to add a changeset to this PR

@vercel

vercel Bot commented Jun 1, 2025

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for Git ↗︎

6 Skipped Deployments
Name Status Preview Updated (UTC)
fermi-contest-2025-02 ⬜️ Ignored (Inspect) Visit Preview Jun 1, 2025 2:38pm
metaforecast ⬜️ Ignored (Inspect) Visit Preview Jun 1, 2025 2:38pm
quri-hub ⬜️ Ignored (Inspect) Visit Preview Jun 1, 2025 2:38pm
quri-ui ⬜️ Ignored (Inspect) Visit Preview Jun 1, 2025 2:38pm
squiggle-components ⬜️ Ignored (Inspect) Visit Preview Jun 1, 2025 2:38pm
squiggle-website ⬜️ Ignored (Inspect) Visit Preview Jun 1, 2025 2:38pm

@dependabot @github

dependabot Bot commented on behalf of github Jul 1, 2025

Copy link
Copy Markdown
Contributor Author

Superseded by #3928.

@dependabot dependabot Bot closed this Jul 1, 2025
@dependabot
dependabot Bot deleted the dependabot/npm_and_yarn/storybook/react-9.0.3 branch July 1, 2025 14:51
@github-project-automation github-project-automation Bot moved this from 🆕 To prioritize to ✅ Done in Overview Jul 1, 2025
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

dependencies Pull requests that update a dependency file

Projects

Status: ✅ Done

Development

Successfully merging this pull request may close these issues.

0 participants