Skip to content

⬆️ Bump @typescript-eslint/parser from 8.29.0 to 8.31.1 - #3809

Closed
dependabot[bot] wants to merge 1 commit into
mainfrom
dependabot/npm_and_yarn/typescript-eslint/parser-8.31.1
Closed

dependabot[bot] wants to merge 1 commit into
mainfrom
dependabot/npm_and_yarn/typescript-eslint/parser-8.31.1

Conversation

@dependabot

@dependabot dependabot Bot commented on behalf of github May 1, 2025

Copy link
Copy Markdown
Contributor

Bumps @typescript-eslint/parser from 8.29.0 to 8.31.1.

Release notes

Sourced from @​typescript-eslint/parser's releases.

v8.31.1

8.31.1 (2025-04-28)

🩹 Fixes

  • eslint-plugin: [no-unnecessary-condition] downgrade fix to suggestion (#11081)

❤️ Thank You

You can read about our versioning strategy and releases on our website.

v8.31.0

8.31.0 (2025-04-21)

🚀 Features

  • eslint-plugin: [no-unnecessary-type-assertion] add option to ignore string const assertions (#10979)

🩹 Fixes

  • typescript-eslint: address bugs in config() around global ignores (#11065)
  • typescript-eslint: gracefully handle invalid flat config objects in config helper (#11070)

❤️ Thank You

You can read about our versioning strategy and releases on our website.

v8.30.1

8.30.1 (2025-04-14)

🚀 Features

  • eslint-plugin: [no-explicit-any] suggest to replace keyof any with PropertyKey (#11032)

🩹 Fixes

  • eslint-plugin: [promise-function-async] use a different error message for functions with promise and non-promise types (#10950)
  • typescript-estree: use token type of PrivateIdentifier instead of Identifier for private identifiers (#11023)
  • eslint-plugin: fix mistake with eslintrc config generation (#11072)

❤️ Thank You

... (truncated)

Changelog

Sourced from @​typescript-eslint/parser's changelog.

8.31.1 (2025-04-28)

This was a version bump only for parser to align it with other projects, there were no code changes.

You can read about our versioning strategy and releases on our website.

8.31.0 (2025-04-21)

This was a version bump only for parser to align it with other projects, there were no code changes.

You can read about our versioning strategy and releases on our website.

8.30.1 (2025-04-14)

This was a version bump only for parser to align it with other projects, there were no code changes.

You can read about our versioning strategy and releases on our website.

8.30.0 (2025-04-14)

This was a version bump only for parser to align it with other projects, there were no code changes.

You can read about our versioning strategy and releases on our website.

8.29.1 (2025-04-07)

This was a version bump only for parser to align it with other projects, there were no code changes.

You can read about our versioning strategy and releases on our website.

Commits

Dependabot compatibility score

Dependabot will resolve any conflicts with this PR as long as you don't alter it yourself. You can also trigger a rebase manually by commenting @dependabot rebase.


Dependabot commands and options

You can trigger Dependabot actions by commenting on this PR:

  • @dependabot rebase will rebase this PR
  • @dependabot recreate will recreate this PR, overwriting any edits that have been made to it
  • @dependabot merge will merge this PR after your CI passes on it
  • @dependabot squash and merge will squash and merge this PR after your CI passes on it
  • @dependabot cancel merge will cancel a previously requested merge and block automerging
  • @dependabot reopen will reopen this PR if it is closed
  • @dependabot close will close this PR and stop Dependabot recreating it. You can achieve the same result by closing it manually
  • @dependabot show <dependency name> ignore conditions will show all of the ignore conditions of the specified dependency
  • @dependabot ignore this major version will close this PR and stop Dependabot creating any more for this major version (unless you reopen the PR or upgrade to it yourself)
  • @dependabot ignore this minor version will close this PR and stop Dependabot creating any more for this minor version (unless you reopen the PR or upgrade to it yourself)
  • @dependabot ignore this dependency will close this PR and stop Dependabot creating any more for this dependency (unless you reopen the PR or upgrade to it yourself)

Bumps [@typescript-eslint/parser](https://github.com/typescript-eslint/typescript-eslint/tree/HEAD/packages/parser) from 8.29.0 to 8.31.1.
- [Release notes](https://github.com/typescript-eslint/typescript-eslint/releases)
- [Changelog](https://github.com/typescript-eslint/typescript-eslint/blob/main/packages/parser/CHANGELOG.md)
- [Commits](https://github.com/typescript-eslint/typescript-eslint/commits/v8.31.1/packages/parser)

---
updated-dependencies:
- dependency-name: "@typescript-eslint/parser"
  dependency-version: 8.31.1
  dependency-type: direct:development
  update-type: version-update:semver-minor
...

Signed-off-by: dependabot[bot] <support@github.com>
@dependabot dependabot Bot added the dependencies Pull requests that update a dependency file label May 1, 2025
@dependabot
dependabot Bot requested review from OAGr and berekuk as code owners May 1, 2025 14:31
@vercel

vercel Bot commented May 1, 2025

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for Git ↗︎

6 Skipped Deployments
Name Status Preview Updated (UTC)
fermi-contest-2025-02 ⬜️ Ignored (Inspect) Visit Preview May 1, 2025 2:38pm
metaforecast ⬜️ Ignored (Inspect) Visit Preview May 1, 2025 2:38pm
quri-hub ⬜️ Ignored (Inspect) Visit Preview May 1, 2025 2:38pm
quri-ui ⬜️ Ignored (Inspect) Visit Preview May 1, 2025 2:38pm
squiggle-components ⬜️ Ignored (Inspect) Visit Preview May 1, 2025 2:38pm
squiggle-website ⬜️ Ignored (Inspect) Visit Preview May 1, 2025 2:38pm

@changeset-bot

changeset-bot Bot commented May 1, 2025

Copy link
Copy Markdown

⚠️ No Changeset found

Latest commit: ad76d4f

Merging this PR will not cause a version bump for any packages. If these changes should not result in a new version, you're good to go. If these changes should result in a version bump, you need to add a changeset.

This PR includes no changesets

When changesets are added to this PR, you'll see the packages that this PR includes changesets for and the associated semver types

Click here to learn what changesets are, and how to add one.

Click here if you're a maintainer who wants to add a changeset to this PR

@github-project-automation github-project-automation Bot moved this to 🆕 To prioritize in Overview May 1, 2025
@dependabot @github

dependabot Bot commented on behalf of github Jun 1, 2025

Copy link
Copy Markdown
Contributor Author

Superseded by #3867.

@dependabot dependabot Bot closed this Jun 1, 2025
@dependabot
dependabot Bot deleted the dependabot/npm_and_yarn/typescript-eslint/parser-8.31.1 branch June 1, 2025 14:58
@github-project-automation github-project-automation Bot moved this from 🆕 To prioritize to ✅ Done in Overview Jun 1, 2025
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

dependencies Pull requests that update a dependency file

Projects

Status: ✅ Done

Development

Successfully merging this pull request may close these issues.

0 participants