Skip to content

fix(deps): update dependency @portabletext/types to ^2.0.15 - #107

Merged
stipsan merged 2 commits into
mainfrom
renovate/portabletext-types-2.x
Aug 25, 2025
Merged

fix(deps): update dependency @portabletext/types to ^2.0.15#107
stipsan merged 2 commits into
mainfrom
renovate/portabletext-types-2.x

Conversation

@renovate

@renovate renovate Bot commented Aug 21, 2025

Copy link
Copy Markdown
Contributor

This PR contains the following updates:

Package Change Age Confidence
@portabletext/types ^2.0.13 -> ^2.0.15 age confidence

Release Notes

portabletext/types (@​portabletext/types)

v2.0.15

Compare Source

Patch Changes

v2.0.14

Compare Source

Patch Changes

Configuration

📅 Schedule: Branch creation - "before 3am on Monday" (UTC), Automerge - At any time (no schedule defined).

🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.

Rebasing: Whenever PR is behind base branch, or you tick the rebase/retry checkbox.

🔕 Ignore: Close this PR and you won't be reminded about this update again.


  • If you want to rebase/retry this PR, check this box

This PR has been generated by Mend Renovate using a curated preset maintained by Sanity. View repository job log here

@changeset-bot

changeset-bot Bot commented Aug 21, 2025

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 46fa9ab

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
@portabletext/toolkit Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@socket-security

socket-security Bot commented Aug 21, 2025

Copy link
Copy Markdown

Review the following changes in direct dependencies. Learn more about Socket for GitHub.

Diff Package Supply Chain
Security
Vulnerability Quality Maintenance License
Updated@​portabletext/​types@​2.0.13 ⏵ 2.0.1591 -81009393 +10100

View full report

@renovate
renovate Bot force-pushed the renovate/portabletext-types-2.x branch 5 times, most recently from 92b8491 to 61a8f15 Compare August 21, 2025 18:47
@renovate renovate Bot changed the title fix(deps): update dependency @portabletext/types to ^2.0.14 fix(deps): update dependency @portabletext/types to ^2.0.15 Aug 21, 2025
@renovate
renovate Bot force-pushed the renovate/portabletext-types-2.x branch 2 times, most recently from 16b4d52 to 84498fe Compare August 25, 2025 10:51
@renovate
renovate Bot force-pushed the renovate/portabletext-types-2.x branch from d40e6a8 to 9c8c880 Compare August 25, 2025 10:53
@stipsan
stipsan enabled auto-merge August 25, 2025 10:59
@stipsan
stipsan added this pull request to the merge queue Aug 25, 2025
Merged via the queue into main with commit 12c0bba Aug 25, 2025
8 checks passed
@stipsan
stipsan deleted the renovate/portabletext-types-2.x branch August 25, 2025 11:00
@ecoscript ecoscript Bot mentioned this pull request Aug 25, 2025
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant