Skip to content

Update vector.yml base to ubi-micro - #12895

Open
jcantrill wants to merge 1 commit into
openshift-eng:logging-6.7from
jcantrill:patch-1
Open

jcantrill wants to merge 1 commit into
openshift-eng:logging-6.7from
jcantrill:patch-1

Conversation

@jcantrill

@jcantrill jcantrill commented Sep 11, 2026

Copy link
Copy Markdown

@openshift-ci openshift-ci Bot added the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Sep 11, 2026
@openshift-ci

openshift-ci Bot commented Sep 11, 2026

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please assign locriandev for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@jcantrill

Copy link
Copy Markdown
Author

@rayfordj is this the correct modification for vector where:

  • the builder image uses rhel9 and needs the rust compiler
  • the final image is to be ubi-micro

@rayfordj

Copy link
Copy Markdown
Contributor

@rayfordj is this the correct modification for vector where:

* the builder image uses rhel9 and needs the rust compiler

* the final image is to be ubi-micro

There should be an entry in images yml for each FROM in your upstream dockerfile. If you want to keep the upstream AS packages change, then we'll need to add an entry for it:

from:
  builder:
    - member: base-rhel9
    - member: base-rhel9
  stream: rhel9-micro

If instead, you remove the AS packages line to again use a single builder, you should be able to change this COPY to:

COPY --from=builder /mnt/rootfs/ /

for clarity, perhaps keep upstream as you have it and add the additional .from.builder.member in ocp-build-data to align? (Suggested change incoming)

Comment thread images/vector.yml
builder:
- member: base-rhel9
member: base-rhel9
stream: rhel9-micro

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Suggested change
stream: rhel9-micro
- member: base-rhel9
stream: rhel9-micro

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants