Repository navigation
43 pnnl example building 1 model 2 broken - #72
Conversation
|
@gtfierro does deploy-book validate all the model files, or is it smart enough to just validate files that have changed as part of the pull request? |
|
I believe it currently rebuilds all of them. A failing model also won't fail the build. If you build locally and look at the built document, that should confirm if the new model is valid. |
|
You can use https://github.com/open223/models.open223.info#validate-or-build-one-model to build/validate a single model, for testing purposes (pull latest from main) |
|
Thanks @gtfierro. Given that all checks passed with no conflicts with base branch, I'm going to merge this PR, close the issue, and see where we are. How long until these changes show up on the open223 website? FYI, I'm going to keep this branch and issue open for now to support additional model updateviterations, if necessary. |
|
@gtfierro how long does it take for a new build to show up on the website?
|
|
Hi @michaelpoplawskipnnl GitHub has been having some issues recently, so there could be some delay. However, in this case, I think the problem is that your updated TTL file does not contain an ontology declaration, which messes up all the tools. The existing tooling has not been super helpful in detecting or reporting these issues, hence why the build passed even though there was still a problem. On my new branch, you can more clearly see the problem: https://github.com/open223/models.open223.info/actions/runs/33193473373/job/98924590728 |
Updated model to latest version: trial 1