Skip to content

Add support for php 8.5 - #15

Merged
hans-thomas merged 3 commits into
2.xfrom
new-patch
Sep 8, 2026
Merged

hans-thomas merged 3 commits into
2.xfrom
new-patch

Conversation

@hans-thomas

Copy link
Copy Markdown
Owner

No description provided.

@hans-thomas hans-thomas self-assigned this Sep 8, 2026
@hans-thomas hans-thomas added the enhancement New feature or request label Sep 8, 2026
@vercel

vercel Bot commented Sep 8, 2026

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for GitHub.

1 Skipped Deployment
Project Deployment Actions Updated
alicia-package Ignored Ignored Sep 8, 2026 2:57pm UTC

@codecov

codecov Bot commented Sep 8, 2026 •

Copy link
Copy Markdown

Codecov Report

✅ All modified and coverable lines are covered by tests.
✅ Project coverage is 86.50%. Comparing base (9f4219e) to head (d9103da).
⚠️ Report is 4 commits behind head on 2.x.

Additional details and impacted files
@@            Coverage Diff            @@
##                2.x      #15   +/-   ##
=========================================
  Coverage     86.50%   86.50%           
  Complexity      127      127           
=========================================
  Files            23       23           
  Lines           489      489           
=========================================
  Hits            423      423           
  Misses           66       66           
Flag Coverage Δ
unittests 86.50% <ø> (?)

Flags with carried forward coverage won't be shown. Click here to find out more.

☔ View full report in Codecov by Harness.
📢 Have feedback on the report? Share it here.

🚀 New features to boost your workflow:
  • ❄️ Test Analytics: Detect flaky tests, report on failures, and find test suite problems.

@hans-thomas
hans-thomas merged commit c80bdc2 into 2.x Sep 8, 2026
17 of 24 checks passed
@hans-thomas
hans-thomas deleted the new-patch branch September 8, 2026 15:04
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

enhancement New feature or request

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant