Skip to content

chore: update mainnet config to match testing - #114

Merged
meyer9 merged 2 commits into
mainfrom
meyer9/update-mainnet-config
Oct 24, 2025
Merged

meyer9 merged 2 commits into
mainfrom
meyer9/update-mainnet-config

Conversation

@meyer9

@meyer9 meyer9 commented Oct 24, 2025

Copy link
Copy Markdown
Collaborator

Description

Testing

@cb-heimdall

cb-heimdall commented Oct 24, 2025 •

Copy link
Copy Markdown
Collaborator

✅ Heimdall Review Status

Requirement Status More Info
Reviews ✅ 1/1
Denominator calculation
Show calculation
1 if user is bot 0
1 if user is external 0
2 if repo is sensitive 0
From .codeflow.yml 1
Additional review requirements
Show calculation
Max 0
0
From CODEOWNERS 0
Global minimum 0
Max 1
1
1 if commit is unverified 0
Sum 1

Comment thread configs/public/mainnet-cross-section.yml Outdated
@meyer9
meyer9 requested a review from tobidae-cb October 24, 2025 18:11
@meyer9
meyer9 enabled auto-merge (squash) October 24, 2025 18:11
@meyer9
meyer9 merged commit 519274b into main Oct 24, 2025
9 checks passed
@meyer9
meyer9 deleted the meyer9/update-mainnet-config branch October 24, 2025 18:28
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants