Skip to content

Lock file maintenance - #996

Merged
renovate-desselbane[bot] merged 1 commit into
masterfrom
renovate/lock-file-maintenance
Sep 28, 2026
Merged

renovate-desselbane[bot] merged 1 commit into
masterfrom
renovate/lock-file-maintenance

Conversation

@renovate-desselbane

Copy link
Copy Markdown
Contributor

This PR contains the following updates:

Update Change
lockFileMaintenance All locks refreshed

🔧 This Pull Request updates lock files to use the latest dependency versions.


Configuration

📅 Schedule: (UTC)

  • Branch creation
    • "before 4am on monday"
  • Automerge
    • At any time (no schedule defined)

🚦 Automerge: Enabled.

♻ Rebasing: Whenever PR is behind base branch, or you tick the rebase/retry checkbox.

👻 Immortal: This PR will be recreated if closed unmerged. Get config help if that's undesired.


  • If you want to rebase/retry this PR, check this box

This PR has been generated by Mend Renovate CLI.

@changeset-bot

changeset-bot Bot commented Sep 28, 2026

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 28c4591

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 7 packages
Name Type
@desselbane/composables Patch
@desselbane/configs Patch
@desselbane/setup Patch
@desselbane/ts-helpers Patch
@desselbane/vitest-helpers Patch
@repo/changelog Patch
@repo/scripts Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@renovate-desselbane
renovate-desselbane Bot merged commit 923036f into master Sep 28, 2026
6 of 7 checks passed
@renovate-desselbane
renovate-desselbane Bot deleted the renovate/lock-file-maintenance branch September 28, 2026 03:58
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

0 participants