Update dependency eslint-plugin-unicorn to v76 - #984
Open
renovate-desselbane[bot] wants to merge 1 commit into
Open
renovate-desselbane[bot] wants to merge 1 commit into
renovate-desselbane[bot] wants to merge 1 commit into
Conversation
🦋 Changeset detectedLatest commit: d79c5dd The changes in this PR will be included in the next version bump. This PR includes changesets to release 1 package
Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
Owner
|
todo reenable no-immediate-mutation |
renovate-desselbane
Bot
force-pushed
the
renovate/major-eslint
branch
10 times, most recently
from
September 27, 2026 16:30
a865f93 to
917ccf8
Compare
| datasource | package | from | to | | ---------- | --------------------- | ------ | ------ | | npm | eslint-plugin-unicorn | 75.0.0 | 76.0.0 |
renovate-desselbane
Bot
force-pushed
the
renovate/major-eslint
branch
from
September 27, 2026 22:51
917ccf8 to
d79c5dd
Compare
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
This PR contains the following updates:
75.0.0→76.0.0Release Notes
sindresorhus/eslint-plugin-unicorn (eslint-plugin-unicorn)
v76.0.0Compare Source
Breaking
no-immediate-mutation: AddcheckConditionalsoption and default it to off (#3707)4686e7eno-break-in-nested-loop: AddcheckContinueoption and default it to off (#3712)d8fbfdeprefer-combined-guards: AddcheckCompoundConditionsoption default it to off (#3710)Improvements
prefer-minimal-ternary: Add autofix (#3716)aaa75a4prefer-logical-operator-over-ternary: Handle boolean expressions (#3711)prefer-ternary: Preserve readable branches (#3709)f7a35aeno-typeof-undefined: Ignore TypeScript ambient variables (#3708)9245ed8db20ce1Configuration
📅 Schedule: (UTC)
🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.
♻ Rebasing: Whenever PR is behind base branch, or you tick the rebase/retry checkbox.
🔕 Ignore: Close this PR and you won't be reminded about this update again.
This PR has been generated by Mend Renovate CLI.